fix bitbucket issues - #3726
Merged
Merged
fix bitbucket issues#3726
Conversation
Signed-off-by: Akshay Gupta <gravity.akshay@gmail.com>
🦋 Changeset detectedLatest commit: fbfecf8 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Contributor
|
|
Contributor
|
Commit SHA:0651ecbc33edc035789c1494f4a1fd94ca220875 Test coverage results 🧪
|
six7
approved these changes
Dec 10, 2025
akshay-gupta7
marked this pull request as ready for review
December 10, 2025 13:03
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Why does this PR exist?
Closes #3725
What does this pull request do?
Testing this change
Additional Notes (if any)