Skip to content

fix bug where generated enum validator type name is incorrect - #200

Merged
joshmossas merged 2 commits into
masterfrom
bugfix/enum-issue
Aug 4, 2026
Merged

fix bug where generated enum validator type name is incorrect#200
joshmossas merged 2 commits into
masterfrom
bugfix/enum-issue

Conversation

@joshmossas

@joshmossas joshmossas commented Aug 4, 2026

Copy link
Copy Markdown
Member

fixes regression in #198

@nx-cloud

nx-cloud Bot commented Aug 4, 2026

Copy link
Copy Markdown

View your CI Pipeline Execution ↗ for commit e927512

Command Status Duration Result
nx build-server test-server-go ✅ Succeeded 20s View ↗
nx build-server test-server-ts ✅ Succeeded 4s View ↗
nx affected -t test ✅ Succeeded 1s View ↗
nx run-many -t compile ✅ Succeeded 4m 21s View ↗
nx build-server test-server-go --configuration=... ✅ Succeeded 26s View ↗
nx build-server test-server-ts --configuration=... ✅ Succeeded 5s View ↗
nx affected -t test --configuration windows-ci ✅ Succeeded <1s View ↗
nx run-many -t compile --configuration windows-ci ✅ Succeeded 1m 53s View ↗
Additional runs (7) ✅ Succeeded ... View ↗

💡 Verify your cache is correct by running tasks in a sandbox. Read docs ↗


☁️ Nx Cloud last updated this comment at 2026-08-04 03:02:37 UTC

@joshmossas
joshmossas merged commit 769f7d7 into master Aug 4, 2026
5 checks passed
@joshmossas
joshmossas deleted the bugfix/enum-issue branch August 4, 2026 03:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant