chore(deps): update react monorepo - #536
Open
renovate[bot] wants to merge 1 commit into
Open
Conversation
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎ 1 Skipped Deployment
|
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
March 3, 2025 15:49
fa4b7c4 to
0fe20c1
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
March 24, 2025 11:34
bf4fcee to
e3b06d5
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
April 2, 2025 10:44
e3b06d5 to
4c771f5
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
April 30, 2025 13:47
41afb46 to
80a3067
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
May 6, 2025 03:27
80a3067 to
27bb7cf
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
May 21, 2025 13:15
36ee201 to
20651c0
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
May 27, 2025 11:25
20651c0 to
0da9e89
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
June 22, 2025 12:46
0da9e89 to
51f51fa
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
August 13, 2025 15:40
fabdbee to
375b8eb
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
August 22, 2025 17:49
375b8eb to
4c3fb00
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
August 31, 2025 12:48
4c3fb00 to
a8ef468
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
September 30, 2025 09:52
9665a16 to
df48da7
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
October 6, 2025 22:07
df48da7 to
87ee745
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
October 21, 2025 10:03
87ee745 to
e5618b1
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
November 18, 2025 08:10
e5618b1 to
930c979
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
December 3, 2025 16:57
930c979 to
938222e
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
January 23, 2026 17:47
16dafe4 to
6f9144a
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
February 5, 2026 13:07
c82cb30 to
199febc
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
February 12, 2026 18:31
199febc to
0d6421a
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
May 22, 2026 00:16
0d6421a to
9fe2447
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
2 times, most recently
from
June 5, 2026 21:09
b3e2adc to
c4218fd
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
August 1, 2026 13:04
c4218fd to
435fde7
Compare
renovate
Bot
force-pushed
the
renovate/react-monorepo
branch
from
September 10, 2026 22:56
435fde7 to
89dffe4
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
18.3.12→18.3.3118.3.1→18.3.7Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
This PR was generated by Mend Renovate. View the repository job log.