chore(deps): update dependency eslint-config-next to v13.5.11 - #535
Open
renovate[bot] wants to merge 1 commit into
Open
chore(deps): update dependency eslint-config-next to v13.5.11#535renovate[bot] wants to merge 1 commit into
renovate[bot] wants to merge 1 commit into
Conversation
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎ 1 Skipped Deployment
|
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
February 11, 2025 00:53
db30c9c to
cb794f1
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
3 times, most recently
from
March 24, 2025 23:47
1fa7750 to
015f571
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
March 27, 2025 03:45
015f571 to
7f359c2
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
2 times, most recently
from
April 8, 2025 23:22
31c9057 to
f2cc336
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
May 20, 2025 19:41
f2cc336 to
0c2f478
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
2 times, most recently
from
June 15, 2025 03:41
a9e535b to
8891005
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
August 10, 2025 13:37
8891005 to
dccb6d9
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
March 27, 2026 01:17
dccb6d9 to
ddafd89
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
March 30, 2026 17:45
ddafd89 to
9aa1845
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
April 27, 2026 19:39
9aa1845 to
df3bb52
Compare
renovate
Bot
force-pushed
the
renovate/nextjs-monorepo
branch
from
April 30, 2026 04:10
df3bb52 to
a320193
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
13.5.7→13.5.11Release Notes
vercel/next.js (eslint-config-next)
v13.5.11Compare Source
Core Changes
v13.5.10Compare Source
Core Changes
v13.5.9Compare Source
Core Changes
v13.5.8Compare Source
Core Changes
d900fadto1dba980: #74202Credits
Huge thanks to @wyattjoh and @ztanner for helping!
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.