Skip to content

chore(deps): update dependency @types/testing-library__jest-dom to v5.14.9 - #949

Open
renovate[bot] wants to merge 1 commit into
developfrom
renovate/testing-library__jest-dom-5.x
Open

chore(deps): update dependency @types/testing-library__jest-dom to v5.14.9#949
renovate[bot] wants to merge 1 commit into
developfrom
renovate/testing-library__jest-dom-5.x

Conversation

@renovate

@renovate renovate Bot commented May 24, 2023

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
@​types/testing-library__jest-dom 5.14.55.14.9 age confidence

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate renovate Bot added dependencies Pull requests that update a dependency file renovate types labels May 24, 2023
@vercel

vercel Bot commented May 24, 2023

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
godwoken-explorer-ui ⬜️ Ignored (Inspect) Visit Preview 💬 Add feedback Aug 10, 2025 3:12pm

@renovate
renovate Bot requested review from Keith-CY and alexsupa597 May 24, 2023 17:11
@renovate renovate Bot changed the title chore(deps): update dependency @types/testing-library__jest-dom to v5.14.6 chore(deps): update dependency @types/testing-library__jest-dom to v5.14.7 Jul 6, 2023
@renovate
renovate Bot force-pushed the renovate/testing-library__jest-dom-5.x branch from 04ca58a to b7267a0 Compare July 6, 2023 00:45
@codecov-commenter

Copy link
Copy Markdown

Codecov Report

Merging #949 (b7267a0) into develop (bc0727a) will decrease coverage by 0.53%.
The diff coverage is n/a.

Impacted file tree graph

@@             Coverage Diff             @@
##           develop     #949      +/-   ##
===========================================
- Coverage    67.60%   67.07%   -0.53%     
===========================================
  Files           18       18              
  Lines         1321     1321              
  Branches       788      788              
===========================================
- Hits           893      886       -7     
- Misses         373      380       +7     
  Partials        55       55              

see 3 files with indirect coverage changes

@renovate renovate Bot changed the title chore(deps): update dependency @types/testing-library__jest-dom to v5.14.7 chore(deps): update dependency @types/testing-library__jest-dom to v5.14.8 Jul 11, 2023
@renovate
renovate Bot force-pushed the renovate/testing-library__jest-dom-5.x branch from b7267a0 to 4e6adf5 Compare July 11, 2023 16:24
@renovate renovate Bot changed the title chore(deps): update dependency @types/testing-library__jest-dom to v5.14.8 chore(deps): update dependency @types/testing-library__jest-dom to v5.14.9 Jul 24, 2023
@renovate
renovate Bot force-pushed the renovate/testing-library__jest-dom-5.x branch from 4e6adf5 to bed0496 Compare July 24, 2023 23:10
@renovate
renovate Bot force-pushed the renovate/testing-library__jest-dom-5.x branch from bed0496 to 1e6b79f Compare August 10, 2025 15:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file renovate types

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant