Skip to content

Bump GCSFuse binary to 3.11.4 - #1623

Merged
yaozile123 merged 1 commit into
GoogleCloudPlatform:mainfrom
yaozile123:bump-gcsfuse-3.11.4
Sep 10, 2026
Merged

yaozile123 merged 1 commit into
GoogleCloudPlatform:mainfrom
yaozile123:bump-gcsfuse-3.11.4

Conversation

@yaozile123

@yaozile123 yaozile123 commented Sep 10, 2026

Copy link
Copy Markdown
Collaborator

What type of PR is this?

Uncomment only one /kind <> line, hit enter to put that in a new line, and remove leading whitespaces from that line:

/kind api-change
/kind bug
/kind cleanup
/kind design
/kind documentation
/kind failing-test

/kind feature

/kind flake

What this PR does / why we need it:
Bump GCSFuse binary to 3.11.4 for cve fixes

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:


@google-oss-prow

Copy link
Copy Markdown

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@gemini-code-assist gemini-code-assist Bot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Code Review

This pull request updates the GCS FUSE binary version from v3.11.3-gke.0 to v3.11.4-gke.0 in the sidecar mounter configuration. There are no review comments, and I have no additional feedback to provide.

@yaozile123
yaozile123 marked this pull request as ready for review September 10, 2026 17:55
@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: chrisThePattyEater, yaozile123

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@yaozile123
yaozile123 merged commit 3b54ecb into GoogleCloudPlatform:main Sep 10, 2026
13 of 15 checks passed
yaozile123 added a commit that referenced this pull request Sep 10, 2026
…23-upstream-release-1.24

Automated cherry pick of #1623: Bump GCSFuse binary to 3.11.4
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants