Skip to content

fix vuln - #1043

Open
sunaydagli wants to merge 1 commit into
masterfrom
vuln
Open

sunaydagli wants to merge 1 commit into
masterfrom
vuln

Conversation

@sunaydagli

Copy link
Copy Markdown
Contributor

No description provided.

@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: sunaydagli
Once this PR has been reviewed and has the lgtm label, please assign paulhong01 for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sunaydagli
sunaydagli requested a review from liusc28 September 28, 2026 07:09
@sunaydagli

Copy link
Copy Markdown
Contributor Author

/retest

@google-oss-prow

Copy link
Copy Markdown

@sunaydagli: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ESPv2-cloud-run-e2e-cloud-function-http-bookstore c9c494f link true /test ESPv2-cloud-run-e2e-cloud-function-http-bookstore
ESPv2-cloud-run-e2e-app-engine-http-bookstore c9c494f link true /test ESPv2-cloud-run-e2e-app-engine-http-bookstore
ESPv2-presubmit-tsan c9c494f link true /test ESPv2-presubmit-tsan
ESPv2-anthos-cloud-run-e2e-anthos-cloud-run-http-bookstore c9c494f link true /test ESPv2-anthos-cloud-run-e2e-anthos-cloud-run-http-bookstore
ESPv2-presubmit-asan c9c494f link true /test ESPv2-presubmit-asan
Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant